Skip to content

chore(deps): update pnpm/action-setup digest to 0977fd9 - #913

Merged
jacobk999 merged 1 commit into
mainfrom
renovate/pnpm-action-setup-digest
Sep 8, 2026
Merged

chore(deps): update pnpm/action-setup digest to 0977fd9#913
jacobk999 merged 1 commit into
mainfrom
renovate/pnpm-action-setup-digest

Conversation

@renovate

@renovate renovate Bot commented Sep 6, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
pnpm/action-setup (changelog) action digest 0ebf4710977fd9

Configuration

📅 Schedule: (UTC)

  • Branch creation
    • "before 12pm on Sunday"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR was generated by Mend Renovate. View the repository job log.

@renovate renovate Bot added the dependencies label Sep 6, 2026
@vercel

vercel Bot commented Sep 6, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
statsify Ignored Ignored Preview Sep 6, 2026 1:00am UTC

@jacobk999
jacobk999 merged commit 6889f4b into main Sep 8, 2026
8 checks passed
@jacobk999
jacobk999 deleted the renovate/pnpm-action-setup-digest branch September 8, 2026 13:23
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant