Skip to content

[release-v1.23] Update Konflux configurations - #1350

Open
serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.23from
serverless-qe:sync-konflux-release-v1.23
Open

serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.23from
serverless-qe:sync-konflux-release-v1.23

Conversation

@serverless-qe

Copy link
Copy Markdown
Collaborator

Update Konflux configurations

@openshift-ci
openshift-ci Bot requested review from aliok and creydr August 24, 2026 10:50
@openshift-ci

openshift-ci Bot commented Aug 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign kaustubh-pande for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 3 times, most recently from d594255 to 4466506 Compare August 31, 2026 13:14
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 15 times, most recently from e6d8178 to 7f0e0db Compare September 7, 2026 11:56
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 4 times, most recently from 26fb129 to 956e114 Compare September 14, 2026 12:17
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 5 times, most recently from 7641d24 to 0a68699 Compare September 16, 2026 15:38
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 5 times, most recently from 6c96ae4 to bf408eb Compare September 18, 2026 04:36
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch from bf408eb to 2271a0b Compare September 18, 2026 10:35
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant