[release-v1.21] Update Konflux configurations - #1903
serverless-qe wants to merge 2 commits into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: serverless-qe The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
b1368a3 to
a561b31
Compare
32ff2f2 to
8c20ade
Compare
2667240 to
4678e40
Compare
994adfb to
e21f57a
Compare
aabbceb to
5aab860
Compare
79a4239 to
3f0d6c5
Compare
f0331cc to
649f75b
Compare
3643bcb to
bffb223
Compare
f56bd78 to
333bea7
Compare
333bea7 to
3271e28
Compare
|
@serverless-qe: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
Update Konflux configurations