Skip to content

[release-1.37] Run make generated-files - #4166

Open
serverless-qe wants to merge 1 commit into
release-1.37from
auto/update-generated-files-release-1.37
Open

serverless-qe wants to merge 1 commit into
release-1.37from
auto/update-generated-files-release-1.37

Conversation

@serverless-qe

Copy link
Copy Markdown
Contributor

Generated by running make generated-files

@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign creydr for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@serverless-qe
serverless-qe force-pushed the auto/update-generated-files-release-1.37 branch from d66af7e to b0e9aad Compare September 15, 2026 13:18
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

@serverless-qe: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/420-kitchensink-upgrade b0e9aad link false /test 420-kitchensink-upgrade
ci/prow/421-upstream-e2e b0e9aad link false /test 421-upstream-e2e
ci/prow/421-mesh-upgrade b0e9aad link false /test 421-mesh-upgrade
ci/prow/421-upstream-e2e-kafka b0e9aad link false /test 421-upstream-e2e-kafka
ci/prow/421-mesh-e2e b0e9aad link false /test 421-mesh-e2e
ci/prow/421-operator-e2e b0e9aad link true /test 421-operator-e2e
ci/prow/421-test-upgrade b0e9aad link true /test 421-test-upgrade
ci/prow/421-kitchensink-e2e b0e9aad link false /test 421-kitchensink-e2e

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants