[release-v1.16] Update Konflux configurations - #1744
Open
serverless-qe wants to merge 1 commit into
Open
serverless-qe wants to merge 1 commit into
serverless-qe wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: serverless-qe The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
4 times, most recently
from
May 7, 2026 09:50
bb4d36a to
9f5c48e
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
7 times, most recently
from
May 21, 2026 06:40
1d47c6a to
329d25e
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
3 times, most recently
from
May 29, 2026 16:49
a7b47c3 to
7a046ff
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
9 times, most recently
from
June 8, 2026 11:10
c734e25 to
81f47dc
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
2 times, most recently
from
June 15, 2026 12:36
a76db60 to
ca6eb8c
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
2 times, most recently
from
June 23, 2026 11:20
083d489 to
6f392a6
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
2 times, most recently
from
August 24, 2026 07:27
56f94c3 to
bf66c0b
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
3 times, most recently
from
August 26, 2026 14:07
597d993 to
57add6a
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
14 times, most recently
from
September 7, 2026 12:01
a932076 to
39780b1
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
4 times, most recently
from
September 14, 2026 12:23
7f86932 to
ed78252
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
5 times, most recently
from
September 16, 2026 15:45
d14c027 to
7a2134e
Compare
serverless-qe
force-pushed
the
sync-konflux-release-v1.16
branch
from
September 16, 2026 18:14
7a2134e to
c48edf0
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Update Konflux configurations