Skip to content

[release-v1.22] Update Konflux configurations - #1764

Open
serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.22from
serverless-qe:sync-konflux-release-v1.22
Open

serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.22from
serverless-qe:sync-konflux-release-v1.22

Conversation

@serverless-qe

Copy link
Copy Markdown
Collaborator

Update Konflux configurations

@openshift-ci
openshift-ci Bot requested review from Fedosin and dsimansk July 20, 2026 04:07
@openshift-ci

openshift-ci Bot commented Jul 20, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign maschmid for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch from cbf66a1 to 14e4d9d Compare July 20, 2026 09:31
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch 2 times, most recently from 3d805bb to ad718d1 Compare August 3, 2026 10:19
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch 2 times, most recently from 4ddd2d5 to a942344 Compare August 17, 2026 07:28
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch 4 times, most recently from 73d1661 to e21860a Compare August 26, 2026 14:07
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch 14 times, most recently from 192f425 to 290e522 Compare September 7, 2026 12:01
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch 4 times, most recently from d1ee111 to 36df0fe Compare September 14, 2026 12:23
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch 5 times, most recently from b7800e9 to f80efa7 Compare September 16, 2026 15:45
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.22 branch from f80efa7 to 6e0a277 Compare September 16, 2026 18:14
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant