Repository navigation
[release-v1.15.x] chore: refactor applyParentUpgrade, correct applyResolutions orderfix and use exact-version resolution entries and discover historical descriptors - #1419
Conversation
…nd use exact-version resolution entries and discover historical descriptors Co-Authored-By: Claude Opus 4.6
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: openshift-cherrypick-robot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
PR Reviewer Guide 🔍Warning
Here are some key observations to aid the review process:
|
|
Important The |
4d09a68
into
openshift-pipelines:release-v1.15.x
This is an automated cherry-pick of #1414
/assign anwesha-palit-redhat