Skip to content

chore(deps): update operator-1-23-bundle to d3af475 - #32973

Closed
red-hat-konflux-kflux-prd-rh02[bot] wants to merge 2 commits into
release-v1.23.xfrom
konflux/component-updates/operator-1-23-index-4-20-component-update-operator-1-23-bundle
Closed

red-hat-konflux-kflux-prd-rh02[bot] wants to merge 2 commits into
release-v1.23.xfrom
konflux/component-updates/operator-1-23-index-4-20-component-update-operator-1-23-bundle

Conversation

@red-hat-konflux-kflux-prd-rh02

Copy link
Copy Markdown
Contributor

Image created from 'https://github.com/openshift-pipelines/operator.git?rev=f253785a96d581541ba4264dfc78635dd13ddb6d'

This PR contains the following updates:

Package Update Change
quay.io/redhat-user-workloads/tekton-ecosystem-tenant/pipelines-operator-bundle digest ba307fa -> d3af475

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test
This is Konflux nudging PR, not Mintmaker.

Image created from 'https://github.com/openshift-pipelines/operator.git?rev=f253785a96d581541ba4264dfc78635dd13ddb6d'

Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot added approved Label for Approved PRs automated For Automated PRs konflux For Konflux PRs konflux-nudge lgtm Looks Good to Me Label labels Oct 8, 2026
@openshift-ci

openshift-ci Bot commented Oct 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@openshift-ci

openshift-ci Bot commented Oct 8, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot removed the lgtm Looks Good to Me Label label Oct 8, 2026
@openshift-ci

openshift-ci Bot commented Oct 8, 2026

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Label for Approved PRs automated For Automated PRs konflux For Konflux PRs konflux-nudge

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant