Skip to content

chore(deps): update konflux references - #3949

Open
red-hat-konflux[bot] wants to merge 1 commit into
release-3.24from
konflux/references/release-3.24
Open

chore(deps): update konflux references#3949
red-hat-konflux[bot] wants to merge 1 commit into
release-3.24from
konflux/references/release-3.24

Conversation

@red-hat-konflux

@red-hat-konflux red-hat-konflux Bot commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
quay.io/konflux-ci/tekton-catalog/task-build-image-index (source, changelog) tekton-bundle digest 290c9ec6ead03c
quay.io/konflux-ci/tekton-catalog/task-buildah-remote-oci-ta (source, changelog) tekton-bundle digest ceecb109afb0e3
quay.io/konflux-ci/tekton-catalog/task-clair-scan (source, changelog) tekton-bundle patch 0.40.4.1
quay.io/konflux-ci/tekton-catalog/task-ecosystem-cert-preflight-checks (source, changelog) tekton-bundle digest c07d2befa53ef4
quay.io/konflux-ci/tekton-catalog/task-git-clone-oci-ta (source, changelog) tekton-bundle digest 2e8fe301d7ba56
quay.io/konflux-ci/tekton-catalog/task-init (source, changelog) tekton-bundle digest 4be9343c6c414a
quay.io/konflux-ci/tekton-catalog/task-prefetch-dependencies-oci-ta (source, changelog) tekton-bundle digest 374f7769b85915
quay.io/konflux-ci/tekton-catalog/task-push-dockerfile-oci-ta (source, changelog) tekton-bundle digest ef00a863e59d63
quay.io/konflux-ci/tekton-catalog/task-source-build-oci-ta (source, changelog) tekton-bundle digest 1808485bbad5bf

Release Notes

konflux-ci/konflux-test-tasks (quay.io/konflux-ci/tekton-catalog/task-clair-scan)

v0.4.1

Changed

Allign script and task version.


Configuration

📅 Schedule: (in timezone Etc/UTC)

  • Branch creation
    • Between 12:00 AM and 07:59 AM (* 0-7 * * *)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

@red-hat-konflux
red-hat-konflux Bot requested a review from rhacs-bot as a code owner September 9, 2026 02:56
@red-hat-konflux
red-hat-konflux Bot requested a review from a team as a code owner September 9, 2026 02:56
@red-hat-konflux red-hat-konflux Bot added the rebuild-test-container Rebuild the collector-tests container. label Sep 9, 2026
@red-hat-konflux
red-hat-konflux Bot requested a review from a team as a code owner September 9, 2026 02:56
@red-hat-konflux
red-hat-konflux Bot enabled auto-merge (squash) September 9, 2026 02:57

@rhacs-bot rhacs-bot left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by automation.

@codecov-commenter

codecov-commenter commented Sep 9, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 27.38%. Comparing base (e00f634) to head (f40cd94).
✅ All tests successful. No failed tests found.

Additional details and impacted files
@@              Coverage Diff              @@
##           release-3.24    #3949   +/-   ##
=============================================
  Coverage         27.38%   27.38%           
=============================================
  Files                95       95           
  Lines              5427     5427           
  Branches           2548     2548           
=============================================
  Hits               1486     1486           
  Misses             3214     3214           
  Partials            727      727           
Flag Coverage Δ
collector-unit-tests 27.38% <ø> (ø)

Flags with carried forward coverage won't be shown. Click here to find out more.

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot force-pushed the konflux/references/release-3.24 branch from c6b10c6 to f40cd94 Compare September 10, 2026 02:18
@github-actions

Copy link
Copy Markdown

/retest collector-on-push

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants