Skip to content

chore(deps): update konflux references - #3884

Merged
red-hat-konflux[bot] merged 1 commit into
masterfrom
konflux/references/master
Sep 23, 2026
Merged

red-hat-konflux[bot] merged 1 commit into
masterfrom
konflux/references/master

Conversation

@red-hat-konflux

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
quay.io/konflux-ci/tekton-catalog/task-clair-scan (source, changelog) tekton-bundle digest 2199648 → c2c9fc4
quay.io/konflux-ci/tekton-catalog/task-clamav-scan (source, changelog) tekton-bundle digest 9df272d → 398514b
quay.io/konflux-ci/tekton-catalog/task-roxctl-scan (source, changelog) tekton-bundle digest 97e2b2c → 4d48db2
quay.io/konflux-ci/tekton-catalog/task-rpms-signature-scan (source, changelog) tekton-bundle digest d9df5d3 → 4b1856e

Configuration

📅 Schedule: (in timezone Etc/UTC)

  • Branch creation
    • Between 12:00 AM and 07:59 AM (* 0-7 * * *)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test.


Documentation

Find out how to configure dependency updates in MintMaker documentation or see all available configuration options in Renovate documentation.

Signed-off-by: red-hat-konflux <126015336+red-hat-konflux[bot]@users.noreply.github.com>
@red-hat-konflux
red-hat-konflux Bot requested review from a team and rhacs-bot as code owners September 23, 2026 03:13
@red-hat-konflux
red-hat-konflux Bot enabled auto-merge (squash) September 23, 2026 03:13

@rhacs-bot rhacs-bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by automation.

@rhacs-bot rhacs-bot left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Auto-approved by automation.

@openshift-ci

openshift-ci Bot commented Sep 23, 2026

Copy link
Copy Markdown

@red-hat-konflux[bot]: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/slim-e2e-tests b39b3af link false /test slim-e2e-tests
ci/prow/e2e-tests b39b3af link false /test e2e-tests

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@red-hat-konflux
red-hat-konflux Bot merged commit 6207caf into master Sep 23, 2026
40 of 42 checks passed
@red-hat-konflux
red-hat-konflux Bot deleted the konflux/references/master branch September 23, 2026 04:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant